Skip to content

logseq: pin electron to electron_39 to fix plugin loading - #515504

Merged
TomaSajt merged 1 commit into
NixOS:masterfrom
TomaSajt:logseq
May 8, 2026
Merged

logseq: pin electron to electron_39 to fix plugin loading#515504
TomaSajt merged 1 commit into
NixOS:masterfrom
TomaSajt:logseq

Conversation

@TomaSajt

@TomaSajt TomaSajt commented May 1, 2026

Copy link
Copy Markdown
Contributor

Closes #506354

Things done

  • Built on platform:
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • Tested, as applicable:
  • Ran nixpkgs-review on this PR. See nixpkgs-review usage.
  • Tested basic functionality of all binary files, usually in ./result/bin/.
  • Nixpkgs Release Notes
    • Package update: when the change is major or breaking.
  • NixOS Release Notes
    • Module addition: when adding a new NixOS module.
    • Module update: when the change is significant.
  • Fits CONTRIBUTING.md, pkgs/README.md, maintainers/README.md and other READMEs.

@nixpkgs-ci nixpkgs-ci Bot added 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 9.needs: reviewer This PR currently has no reviewers requested and needs attention. 2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". labels May 1, 2026

@niklaskorz niklaskorz left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Electron 39 is EOL tomorrow

https://endoflife.date/electron

Edit: I see you're already aware of this from the issue comment. It's not obvious to me what Electron version the nightly version is on, but if that supports a secure version of Electron, temporarily updating the package to an unstable version would be preferable over keeping an insecure dependency IMHO. Feel free to dismiss this review if you prefer sticking to Electron 39.

@nixpkgs-ci nixpkgs-ci Bot removed the 9.needs: reviewer This PR currently has no reviewers requested and needs attention. label May 4, 2026
@TomaSajt

TomaSajt commented May 4, 2026

Copy link
Copy Markdown
Contributor Author

Upstream has gone through a sizable refactor since the last release, making a quick update to unstable difficult. However, upstream is still using electron_39 in the latest commit, so my guess is that the issue is still not fixed. I'll try it with electron_40 when I manage to compile the latest revision.
In the meantime, I think a working but insecure solution is acceptable, since people will be manually pinning to electron_39 anyways.

@TomaSajt

TomaSajt commented May 5, 2026

Copy link
Copy Markdown
Contributor Author

Btw I managed to build the latest revision (#516682), but it still doesn't work with Electron 40+

@teutat3s
teutat3s self-requested a review May 6, 2026 09:01
@TomaSajt

TomaSajt commented May 8, 2026

Copy link
Copy Markdown
Contributor Author

I'm gonna merge this for now and we'll figure out a solution later.

@TomaSajt
TomaSajt added this pull request to the merge queue May 8, 2026
Merged via the queue into NixOS:master with commit 553ce73 May 8, 2026
38 of 40 checks passed
@TomaSajt
TomaSajt deleted the logseq branch May 8, 2026 09:19
@TomaSajt TomaSajt mentioned this pull request May 16, 2026
3 tasks
@nixpkgs-ci

nixpkgs-ci Bot commented May 16, 2026

Copy link
Copy Markdown
Contributor

Successfully created backport PR for release-25.11:

@github-actions github-actions Bot added the 8.has: port to stable This PR already has a backport to the stable release. label May 16, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". 8.has: port to stable This PR already has a backport to the stable release. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Logseq: Hangs with all plugins failing to load on nixpkgs unstable

2 participants