Skip to content

imgui: 1.91.4 -> 1.92.7 - #505089

Open
jemand771 wants to merge 3 commits into
NixOS:masterfrom
jemand771:imgui-1.92.6
Open

jemand771 wants to merge 3 commits into
NixOS:masterfrom
jemand771:imgui-1.92.6

Conversation

@jemand771

@jemand771 jemand771 commented Mar 30, 2026

Copy link
Copy Markdown
Member

Changelog: https://github.com/ocornut/imgui/releases
Diff: https://github.com/ocornut/imgui/compare/v1.91.4..v1.92.7

Update imgui to v1.92.7 and fix any breakage caused by it. Included patch for dartsim is merged by upstream but still unreleased, patch for imtui is not merged and so far unreviewed. Maybe someone from nixpkgs can take a quick look there: ggerganov/imtui#62

Supersedes:

Depends on (wait for merge):

ignore my branch name, 1.92.7 came out while I was waiting for other things to settle / get merged

Things done

  • Built on platform:
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • Tested, as applicable:
  • Ran nixpkgs-review on this PR. See nixpkgs-review usage.
  • Tested basic functionality of all binary files, usually in ./result/bin/.
  • Nixpkgs Release Notes
    • Package update: when the change is major or breaking.
  • NixOS Release Notes
    • Module addition: when adding a new NixOS module.
    • Module update: when the change is significant.
  • Fits CONTRIBUTING.md, pkgs/README.md, maintainers/README.md and other READMEs.

@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (update) This PR updates a package to a newer version 10.rebuild-linux: 11-100 This PR causes between 11 and 100 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. labels Mar 30, 2026
@jemand771 jemand771 changed the title imgui: 1.91.4 -> 1.92.6 imgui: 1.91.4 -> 1.92.7 May 1, 2026
@jemand771

Copy link
Copy Markdown
Member Author

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review pr 505089
Commit: e8cc6fb26694478936fbd0990df3cd8214572d36


x86_64-linux

✅ 14 packages built:
  • box2d
  • comaps
  • dartsim
  • imgui
  • imgui-file-dialog
  • imgui-file-dialog.dev
  • imgui.lib
  • imnodes
  • imnodes.dev
  • implot
  • imtui
  • python313Packages.dartsim
  • python314Packages.dartsim
  • sony-headphones-client

@jemand771
jemand771 marked this pull request as ready for review May 1, 2026 14:38
@jemand771
jemand771 requested review from SomeoneSerge and nim65s May 1, 2026 14:46

@nim65s nim65s left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

we should add ?full_index=1 to github .patch urls: https://github.com/NixOS/nixpkgs/blob/master/pkgs/README.md#fetching-patches

otherwise diff lgtm

@jemand771

Copy link
Copy Markdown
Member Author

@nim65s as far as I understand it, that only applies to fetchpatch2, not fetchpatch.
on that note, I'm not really sure when to use which one (the 2 seems to imply "newer" and "better", but it's used almost nowhere) - I can switch over (and add full_index=1), if there's a reason to

@nim65s

nim65s commented May 5, 2026

Copy link
Copy Markdown
Contributor

Yes, the situation is not super clear. I guess we can dismiss my comment here

@voidlily voidlily mentioned this pull request May 12, 2026
13 tasks
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

8.has: package (update) This PR updates a package to a newer version 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-linux: 11-100 This PR causes between 11 and 100 packages to rebuild on Linux.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants