Skip to content

imgui: 1.91.4 -> 1.92.3 - #444588

Open
Rucadi wants to merge 1 commit into
NixOS:masterfrom
Rucadi:imgui
Open

Rucadi wants to merge 1 commit into
NixOS:masterfrom
Rucadi:imgui

Conversation

@Rucadi

@Rucadi Rucadi commented Sep 20, 2025

Copy link
Copy Markdown
Contributor

Probably this PR is WIP.

For some reason it failed to build glfw without adding as native deps xorg and wayland, this was not needed before and I have no idea why.

There are almost 400 dependent packages and imgui has had api changes, no idea yet what packages are affected.

I create this but first we should understand if this is safe to merge or not

Things done

  • Built on platform:
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • Tested, as applicable:
  • Ran nixpkgs-review on this PR. See nixpkgs-review usage.
  • Tested basic functionality of all binary files, usually in ./result/bin/.
  • Nixpkgs Release Notes
    • Package update: when the change is major or breaking.
  • NixOS Release Notes
    • Module addition: when adding a new NixOS module.
    • Module update: when the change is significant.
  • Fits CONTRIBUTING.md, pkgs/README.md, maintainers/README.md and other READMEs.

Add a 👍 reaction to pull requests you find important.

@nixpkgs-ci nixpkgs-ci Bot added 10.rebuild-linux: 11-100 This PR causes between 11 and 100 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. labels Sep 20, 2025
@nix-owners
nix-owners Bot requested a review from SomeoneSerge September 20, 2025 10:27
@Rucadi

Rucadi commented Sep 20, 2025

Copy link
Copy Markdown
Contributor Author

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review pr 444588
Commit: 6f44f6ce0b9edcd85dffbb9c1c12ea747c537f7a


x86_64-linux

❌ 4 packages failed to build:
  • imnodes
  • imnodes.dev
  • implot
  • imtui
✅ 10 packages built:
  • _2ship2harkinian
  • box2d
  • cemu
  • comaps
  • dartsim
  • imgui
  • imgui.lib
  • python312Packages.dartsim
  • python313Packages.dartsim
  • sony-headphones-client

Error logs: `x86_64-linux`
imnodes

-- Build files have been written to: /build/source/build
cmake: enabled parallel building
cmake: enabled parallel installing
Running phase: buildPhase
@nix { "action": "setPhase", "phase": "buildPhase" }
build flags: -j16 SHELL=/nix/store/q7sqwn7i6w2b67adw0bmix29pxg85x3w-bash-5.3p3/bin/bash
[ 50%] Building CXX object CMakeFiles/imnodes.dir/imnodes.cpp.o
/build/source/imnodes.cpp: In function 'void ImNodes::{anonymous}::ImDrawListGrowChannels(ImDrawList*, int)':
/build/source/imnodes.cpp:353:22: error: 'struct ImDrawCmd' has no member named 'TextureId'
353 | draw_cmd.TextureId = draw_list->_TextureIdStack.back();
| ^~~~~~~~~
/build/source/imnodes.cpp:353:45: error: 'struct ImDrawList' has no member named '_TextureIdStack'; did you mean '_TextureStack'?
353 | draw_cmd.TextureId = draw_list->_TextureIdStack.back();
| ^~~~~~~~~~~~~~~
| _TextureStack
make[2]: *** [CMakeFiles/imnodes.dir/build.make:79: CMakeFiles/imnodes.dir/imnodes.cpp.o] Error 1
make[1]: *** [CMakeFiles/Makefile2:87: CMakeFiles/imnodes.dir/all] Error 2
make: *** [Makefile:136: all] Error 2

implot
  872 |         ImPlot::PlotImage("my image",ImGui::GetIO().Fonts->TexID, bmin, bmax, uv0, uv1, tint);
      |                                      ~~~~~~~~~~~~~~~~~~~~~~^~~~~
      |                                                            |
      |                                                            ImTextureRef
In file included from /build/source/implot_demo.cpp:30:
/build/source/implot.h:912:61: note:   initializing argument 2 of 'void ImPlot::PlotImage(const char*, ImTextureID, const ImPlotPoint&, const ImPlotPoint&, const ImVec2&, const ImVec2&, const ImVec4&, ImPlotImageFlags)'
  912 | IMPLOT_API void PlotImage(const char* label_id, ImTextureID user_texture_id, const ImPlotPoint& bounds_min, const ImPlotPoint& bounds_max, const ImVec2& uv0=ImVec2(0,0), const ImVec2& uv1=ImVec2(1,1), const ImVec4& tint_col=ImVec4(1,1,1,1), ImPlotImageFlags flags=0);
      |                                                 ~~~~~~~~~~~~^~~~~~~~~~~~~~~
/build/source/implot.cpp: In function 'void ImPlot::AddTextVertical(ImDrawList*, ImVec2, ImU32, const char*, const char*)':
/build/source/implot.cpp:345:44: error: 'struct ImFont' has no member named 'FontSize'
  345 |     const float scale = g.FontSize / font->FontSize;
      |                                            ^~~~~~~~
/build/source/implot.cpp:362:43: error: 'struct ImFont' has no member named 'FindGlyph'
  362 |         const ImFontGlyph * glyph = font->FindGlyph((ImWchar)c);
      |                                           ^~~~~~~~~
make[2]: *** [CMakeFiles/implot.dir/build.make:107: CMakeFiles/implot.dir/implot_demo.cpp.o] Error 1
make[2]: *** Waiting for unfinished jobs....
make[2]: *** [CMakeFiles/implot.dir/build.make:79: CMakeFiles/implot.dir/implot.cpp.o] Error 1
make[1]: *** [CMakeFiles/Makefile2:87: CMakeFiles/implot.dir/all] Error 2
make: *** [Makefile:136: all] Error 2
imtui
  266 |     if (ImGui::GetIO().KeysDown[ImGui::GetIO().KeyMap[ImGuiKey_X]]) ImGui::GetIO().KeyCtrl = true;
      |                                                ^~~~~~
/build/source/src/imtui-impl-ncurses.cpp:267:24: error: 'struct ImGuiIO' has no member named 'KeysDown'
  267 |     if (ImGui::GetIO().KeysDown[ImGui::GetIO().KeyMap[ImGuiKey_Y]]) ImGui::GetIO().KeyCtrl = true;
      |                        ^~~~~~~~
/build/source/src/imtui-impl-ncurses.cpp:267:48: error: 'struct ImGuiIO' has no member named 'KeyMap'
  267 |     if (ImGui::GetIO().KeysDown[ImGui::GetIO().KeyMap[ImGuiKey_Y]]) ImGui::GetIO().KeyCtrl = true;
      |                                                ^~~~~~
/build/source/src/imtui-impl-ncurses.cpp:268:24: error: 'struct ImGuiIO' has no member named 'KeysDown'
  268 |     if (ImGui::GetIO().KeysDown[ImGui::GetIO().KeyMap[ImGuiKey_Z]]) ImGui::GetIO().KeyCtrl = true;
      |                        ^~~~~~~~
/build/source/src/imtui-impl-ncurses.cpp:268:48: error: 'struct ImGuiIO' has no member named 'KeyMap'
  268 |     if (ImGui::GetIO().KeysDown[ImGui::GetIO().KeyMap[ImGuiKey_Z]]) ImGui::GetIO().KeyCtrl = true;
      |                                                ^~~~~~
[3/10] Building CXX object third-party/CMakeFiles/imgui-for-imtui.dir/imgui/imgui/imgui_tables.cpp.o
[4/10] Building CXX object third-party/CMakeFiles/imgui-for-imtui.dir/imgui/imgui/imgui_demo.cpp.o
[5/10] Building CXX object third-party/CMakeFiles/imgui-for-imtui.dir/imgui/imgui/imgui_draw.cpp.o
[6/10] Building CXX object third-party/CMakeFiles/imgui-for-imtui.dir/imgui/imgui/imgui_widgets.cpp.o
[7/10] Building CXX object third-party/CMakeFiles/imgui-for-imtui.dir/imgui/imgui/imgui.cpp.o
ninja: build stopped: subcommand failed.

Comment on lines +97 to +103
nativeBuildInputs = [
cmake
xorg.libX11
xorg.libXcursor
xorg.libXrandr
wayland
];

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

For some reason it failed to build glfw without adding as native deps xorg and wayland, this was not needed before and I have no idea why.

Huh, native specifically, not buildInputs? For wayland it could be wayland-scanner maybe, but the rest are supposed to be... just libraries?

Could I ask you to to publish the build logs for the failing version please? A gist e.g.

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Please move them to buildInputs

@SomeoneSerge

Copy link
Copy Markdown
Contributor

imgui has had api changes, no idea yet what packages are affected.

Neighbour projects like implot and imnodes are probably fixed by updating them too. The CI label so far says <100 rebuilds, not 400, I guess that's good news:)

@nixpkgs-ci nixpkgs-ci Bot added the 2.status: merge conflict This PR has merge conflicts with the target branch label Feb 6, 2026
@nixpkgs-ci nixpkgs-ci Bot added the 2.status: stale https://github.com/NixOS/nixpkgs/blob/master/.github/STALE-BOT.md label Mar 25, 2026
@jemand771 jemand771 mentioned this pull request Mar 30, 2026
13 tasks
@SuperSandro2000 SuperSandro2000 changed the title Imgui: 1.91.4 -> 1.92.3 imgui: 1.91.4 -> 1.92.3 Apr 8, 2026
@nixpkgs-ci nixpkgs-ci Bot removed the 2.status: stale https://github.com/NixOS/nixpkgs/blob/master/.github/STALE-BOT.md label Apr 8, 2026
@voidlily

Copy link
Copy Markdown
Member

Close in favor of #505089 (1.92.7)?

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

2.status: merge conflict This PR has merge conflicts with the target branch 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-linux: 11-100 This PR causes between 11 and 100 packages to rebuild on Linux.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants