Skip to content

auto-detect main modules - #99

Merged
k8s-ci-robot merged 1 commit into
kubernetes-sigs:mainfrom
dims:main
Feb 8, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes-sigs:mainfrom
dims:main

Conversation

@dims

@dims dims commented Feb 8, 2026

Copy link
Copy Markdown
Member

Auto-detect main modules from go.work or recursive go.mod scan; exclude tools by default and print detected modules.

Signed-off-by: Davanum Srinivas <davanum@gmail.com>
@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Feb 8, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dims

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Feb 8, 2026
@dims dims added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 8, 2026
@k8s-ci-robot
k8s-ci-robot merged commit d8defea into kubernetes-sigs:main Feb 8, 2026
7 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants