Skip to content

Enhancement/add duplicate prog test - #1910

Open
R4R3D1FF wants to merge 7 commits into
kmesh-net:mainfrom
R4R3D1FF:enhancement/add-duplicate-prog-test
Open

R4R3D1FF wants to merge 7 commits into
kmesh-net:mainfrom
R4R3D1FF:enhancement/add-duplicate-prog-test

Conversation

@R4R3D1FF

@R4R3D1FF R4R3D1FF commented Aug 9, 2026

Copy link
Copy Markdown

What type of PR is this?

/kind enhancement

What this PR does / why we need it:

Adds tests for kmesh restart ensuring no duplicate program is present.

Which issue(s) this PR fixes:
Fixes #799

Special notes for your reviewer:

Does this PR introduce a user-facing change?:

NONE

@kmesh-bot kmesh-bot added the kind/enhancement New feature or request label Aug 9, 2026
@kmesh-bot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign lizhencheng9527 for approval. For more information see the Kubernetes Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@yashisrani

Copy link
Copy Markdown
Contributor

@R4R3D1FF ci are failing, so please fix it ..

@R4R3D1FF
R4R3D1FF marked this pull request as draft August 10, 2026 11:46
@R4R3D1FF
R4R3D1FF force-pushed the enhancement/add-duplicate-prog-test branch from 37c93c9 to bcf02e8 Compare August 10, 2026 16:35
@codecov

codecov Bot commented Aug 10, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 39.53%. Comparing base (32fe3df) to head (0de4347).
⚠️ Report is 18 commits behind head on main.
see 4 files with indirect coverage changes


Continue to review full report in Codecov by Harness.

Legend - Click here to learn more
Δ = absolute <relative> (impact), ø = not affected, ? = missing data
Powered by Codecov. Last update 045678d...0de4347. Read the comment docs.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@R4R3D1FF
R4R3D1FF marked this pull request as ready for review August 11, 2026 17:03
@R4R3D1FF

Copy link
Copy Markdown
Author

@R4R3D1FF ci are failing, so please fix it ..

Should work now @yashisrani

Comment thread pkg/bpf/utils/bpf_helper_test.go Outdated
}

cgopt := link.CgroupOptions{
Path: "/sys/fs/cgroup",

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Shouldn't it be cgroup2?

Copy link
Copy Markdown
Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I don't think it matters for the focus of this test since it's only supposed to test whether BpfProgUpdate properly replaces the program so that it is removed after no link points to it.

Should I still replace it anyway? (even though it has no actual affect on the functioning of the test)

Signed-off-by: R4R3D1FF <mausam191003@gmail.com>
…sh restart

Signed-off-by: R4R3D1FF <mausam191003@gmail.com>
Signed-off-by: R4R3D1FF <mausam191003@gmail.com>
Signed-off-by: R4R3D1FF <mausam191003@gmail.com>
… not already mounted

Signed-off-by: R4R3D1FF <mausam191003@gmail.com>
Signed-off-by: R4R3D1FF <mausam191003@gmail.com>
@R4R3D1FF
R4R3D1FF force-pushed the enhancement/add-duplicate-prog-test branch from 0de4347 to 78f4ad4 Compare August 31, 2026 08:44
Signed-off-by: R4R3D1FF <mausam191003@gmail.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/enhancement New feature or request size/L

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Add restart test

4 participants