Skip to content

build(deps): bump py-cov-action/python-coverage-comment-action from 4.4 to 4.5 - #13004

Merged
bogdankostic merged 1 commit into
mainfrom
dependabot/github_actions/py-cov-action/python-coverage-comment-action-4.5
Sep 29, 2026
Merged

bogdankostic merged 1 commit into
mainfrom
dependabot/github_actions/py-cov-action/python-coverage-comment-action-4.5

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 28, 2026

Copy link
Copy Markdown
Contributor

Bumps py-cov-action/python-coverage-comment-action from 4.4 to 4.5.

Release notes

Sourced from py-cov-action/python-coverage-comment-action's releases.

v4.5

What's Changed

Full Changelog: py-cov-action/python-coverage-comment-action@v4.4...v4.5

Commits

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [py-cov-action/python-coverage-comment-action](https://github.com/py-cov-action/python-coverage-comment-action) from 4.4 to 4.5.
- [Release notes](https://github.com/py-cov-action/python-coverage-comment-action/releases)
- [Commits](py-cov-action/python-coverage-comment-action@54b4ecb...e5fb221)

---
updated-dependencies:
- dependency-name: py-cov-action/python-coverage-comment-action
  dependency-version: '4.5'
  dependency-type: direct:production
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file github_actions Pull requests that update GitHub Actions code labels Sep 28, 2026
@dependabot
dependabot Bot requested a review from a team as a code owner September 28, 2026 14:27
@dependabot
dependabot Bot requested review from bogdankostic and removed request for a team September 28, 2026 14:27
@dependabot dependabot Bot added github_actions Pull requests that update GitHub Actions code dependencies Pull requests that update a dependency file labels Sep 28, 2026
@vercel

vercel Bot commented Sep 28, 2026 •

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
haystack-docs Ignored Ignored Preview Sep 28, 2026 2:27pm UTC

Request Review

@bogdankostic
bogdankostic merged commit 4bb6599 into main Sep 29, 2026
39 checks passed
@bogdankostic
bogdankostic deleted the dependabot/github_actions/py-cov-action/python-coverage-comment-action-4.5 branch September 29, 2026 07:56
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file github_actions Pull requests that update GitHub Actions code topic:CI

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant