Skip to content

docs: sync Core Integrations API reference (astra) on Docusaurus - #12960

Merged
github-actions[bot] merged 1 commit into
mainfrom
sync-docusaurus-api-reference-astra
Sep 25, 2026
Merged

github-actions[bot] merged 1 commit into
mainfrom
sync-docusaurus-api-reference-astra

Conversation

@HaystackBot

Copy link
Copy Markdown
Contributor

This PR syncs the Core Integrations API reference (astra) on Docusaurus. Just approve and merge it.

@vercel

vercel Bot commented Sep 25, 2026 •

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
haystack-docs Ignored Ignored Preview Sep 25, 2026 2:30pm UTC

Request Review

@github-actions
github-actions Bot enabled auto-merge (squash) September 25, 2026 14:30
@github-actions
github-actions Bot merged commit f6b8b07 into main Sep 25, 2026
22 checks passed
@github-actions
github-actions Bot deleted the sync-docusaurus-api-reference-astra branch September 25, 2026 14:32
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants