Skip to content

[Core] - Replace the 'glob' package with Node.js internal method - #102

Merged
amaify merged 2 commits into
devfrom
tobe/reduce-bundle-size
Aug 10, 2026
Merged

[Core] - Replace the 'glob' package with Node.js internal method#102
amaify merged 2 commits into
devfrom
tobe/reduce-bundle-size

Conversation

@ghost

@ghost ghost commented Aug 10, 2026

Copy link
Copy Markdown

No description provided.

@ghost
ghost requested a review from amaify August 10, 2026 14:52
@ghost ghost self-assigned this Aug 10, 2026
@coderabbitai

coderabbitai Bot commented Aug 10, 2026

Copy link
Copy Markdown

Note

Currently processing new changes in this PR. This may take a few minutes, please wait...

⚙️ Run configuration

Configuration used: defaults

Review profile: CHILL

Plan: Pro Plus

Run ID: e9b87039-dbc3-4dd7-a263-535174d7b255

📥 Commits

Reviewing files that changed from the base of the PR and between 9fd9d91 and 5f7de0b.

⛔ Files ignored due to path filters (1)
  • pnpm-lock.yaml is excluded by !**/pnpm-lock.yaml
📒 Files selected for processing (18)
  • CHANGELOG.md
  • package.json
  • src/cli/index.ts
  • src/core/get-setup-function.ts
  • src/core/test/get-setup-function.test.ts
  • src/utils/wallets/get-wallet-extension-id-from-browser.ts
  • src/wallets/keplr/actions/get-account-address.keplr.ts
  • src/wallets/keplr/actions/rename-account.keplr.ts
  • src/wallets/keplr/types.ts
  • src/wallets/metamask/actions/add-account.metamask.ts
  • src/wallets/metamask/actions/add-custom-network.metamask.ts
  • src/wallets/metamask/types.ts
  • src/wallets/petra/actions/rename-account.petra.ts
  • src/wallets/phantom/actions/get-account-address.phantom.ts
  • src/wallets/phantom/actions/switch-account.phantom.ts
  • src/wallets/solflare/actions/add-account.solflare.ts
  • src/wallets/solflare/types.ts
  • tsup.config.ts
 ______________________________________________________________________
< You used `any` like it's a life jacket. Spoiler: it's a pool noodle. >
 ----------------------------------------------------------------------
  \
   \   \
        \ /\
        ( )
      .( o ).
✨ Finishing Touches
📝 Generate docstrings
  • Create stacked PR
  • Commit on current branch
🧪 Generate unit tests (beta)
  • Create PR with unit tests
  • Commit unit tests in branch tobe/reduce-bundle-size

Comment @coderabbitai help to get the list of available commands.

@amaify
amaify merged commit 1c3b288 into dev Aug 10, 2026
6 of 7 checks passed
@amaify
amaify deleted the tobe/reduce-bundle-size branch August 10, 2026 15:01
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant